Skip to content

feat: make params able to be number of boolean (stringable) - #21

Merged
yoannfleurydev merged 2 commits into
mainfrom
feat/stringable-parameters
Mar 30, 2026
Merged

yoannfleurydev merged 2 commits into
mainfrom
feat/stringable-parameters

Conversation

@yoannfleurydev

Copy link
Copy Markdown
Member

closes #20

@changeset-bot

changeset-bot Bot commented Mar 30, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: fd6cafe

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@bearstudio/lunalink Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@yoannfleurydev
yoannfleurydev merged commit e7bc62d into main Mar 30, 2026
1 check passed
@yoannfleurydev
yoannfleurydev deleted the feat/stringable-parameters branch March 30, 2026 10:16
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Should path params auto-cast values to string?

1 participant