chore(deps): update dependency i18next-parser to v8.13.0 - #111
Open
renovate[bot] wants to merge 1 commit into
Open
chore(deps): update dependency i18next-parser to v8.13.0#111renovate[bot] wants to merge 1 commit into
renovate[bot] wants to merge 1 commit into
Conversation
|
The latest updates on your projects. Learn more about Vercel for GitHub.
1 Skipped Deployment
|
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
July 4, 2024 04:07
3bc5a1b to
9f9417c
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
July 5, 2024 05:36
9f9417c to
a4d7d7e
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
July 26, 2024 02:45
a4d7d7e to
1d32156
Compare
Keith-CY
force-pushed
the
master
branch
4 times, most recently
from
August 16, 2024 14:01
6cec327 to
cd9ae83
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
January 23, 2025 23:20
1d32156 to
8dd212c
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
January 30, 2025 18:26
8dd212c to
92b8983
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
April 1, 2025 13:51
cdd1d37 to
19ece18
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
May 19, 2025 16:44
19ece18 to
5abce2f
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
May 28, 2025 09:55
5abce2f to
44e7d6d
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
June 4, 2025 12:07
44e7d6d to
58d7dfd
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
June 22, 2025 15:39
58d7dfd to
d6eb34b
Compare
renovate
Bot
force-pushed
the
renovate/i18next-parser-8.x-lockfile
branch
from
August 10, 2025 14:20
d6eb34b to
729034c
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
8.7.0→8.13.0Release Notes
i18next/i18next-parser (i18next-parser)
v8.13.0Compare Source
v8.12.0Compare Source
v8.11.0Compare Source
v8.10.0Compare Source
v8.9.0Compare Source
v8.8.0Compare Source
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.