python3Packages.update-dotdee: modernize and migrate to pyproject and add changelog - #528097
Merged
Sigmanificient merged 2 commits intoJun 21, 2026
Conversation
Sigmanificient
approved these changes
Jun 4, 2026
Sigmanificient
left a comment
Member
There was a problem hiding this comment.
diff lgtm, work with python3.12, but a deps is blcoking for 3.13+
executor-23.2 not supported for interpreter python3.13
Contributor
Author
That is fixed with #528079 |
Member
|
Error updating the branch: GraphQL: user doesn't have permission to update head repository 🤔 |
Ambossmann
force-pushed
the
pyproject-migration/python3Packages.update-dotdee
branch
from
June 20, 2026 09:48
de2b971 to
30d128c
Compare
Ambossmann
force-pushed
the
pyproject-migration/python3Packages.update-dotdee
branch
from
June 20, 2026 09:51
30d128c to
fdb12c0
Compare
Contributor
Author
I rebased it onto master in case that was what you were trying to do. |
Member
|
eyJhb
approved these changes
Jun 20, 2026
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
formattopyproject#515974Things done
passthru.tests.nixpkgs-reviewon this PR. See nixpkgs-review usage../result/bin/.