Skip to content

scripts: project-intel: Remove Agilex build artifacts on clean - #2098

Open
bluncan wants to merge 1 commit into
mainfrom
intel_makefile
Open

bluncan wants to merge 1 commit into
mainfrom
intel_makefile

Conversation

@bluncan

@bluncan bluncan commented Jul 3, 2026

Copy link
Copy Markdown
Contributor

PR Description

When building Agilex projects, Quartus generates a lot of artifacts that were not being handled by the old Makefile.

PR Type

  • Bug fix (change that fixes an issue)
  • New feature (change that adds new functionality)
  • Breaking change (has dependencies in other repos or will cause CI to fail)
  • Documentation

PR Checklist

  • I have followed the code style guidelines
  • I have performed a self-review of changes
  • I have compiled all hdl projects and libraries affected by this PR
  • I have tested in hardware affected projects, at least on relevant boards
  • I have commented my code, at least hard-to-understand parts
  • I have signed off all commits from this PR
  • I have updated the documentation (wiki pages, ReadMe files, Copyright etc)
  • I have not introduced new Warnings/Critical Warnings on compilation
  • I have added new hdl testbenches or updated existing ones

Signed-off-by: Bogdan Luncan <bogdan.luncan@analog.com>
Comment thread .gitignore
*.legacy
*.info
*.tcl.history*
sm_hssi_port_rotation_param_map.json No newline at end of file

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Missing newline character at the end

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants