Skip to content
This repository was archived by the owner on Aug 31, 2026. It is now read-only.

Strip functionality from old site - #5933

Merged
uiuxxx merged 3 commits into
deployfrom
strip-functionality
Feb 3, 2026
Merged

uiuxxx merged 3 commits into
deployfrom
strip-functionality

Conversation

@uiuxxx

@uiuxxx uiuxxx commented Feb 3, 2026 •

Copy link
Copy Markdown
Contributor
  • Remove "Claims" from navigation
  • Remove veBAL voting table
  • Replace veBAL button actions with redirects to the veBAL manage page on balancer.fi
  • Remove deprecated links, like Grants and Feedback

@uiuxxx uiuxxx changed the title Remove links from nav, including Claims and Feedback Strip functionality from old site Feb 3, 2026
@vercel

vercel Bot commented Feb 3, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
app-v2 Ignored Ignored Feb 3, 2026 4:18pm

Request Review

@uiuxxx
uiuxxx merged commit 0e4f8e4 into deploy Feb 3, 2026
5 checks passed
@uiuxxx
uiuxxx deleted the strip-functionality branch February 3, 2026 16:41
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants