Skip to content

MGMT-12656: CIM: Append table actions to content-scrollable element - #2069

Closed
rawagner wants to merge 1 commit into
openshift-assisted:masterfrom
rawagner:cim_table_act
Closed

MGMT-12656: CIM: Append table actions to content-scrollable element#2069
rawagner wants to merge 1 commit into
openshift-assisted:masterfrom
rawagner:cim_table_act

Conversation

@rawagner

Copy link
Copy Markdown
Member

This will work only on OCP 4.13 as we need latest pf/react-table

@rawagner
rawagner requested a review from a team as a code owner April 28, 2023 11:23
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 28, 2023
@openshift-ci-robot

openshift-ci-robot commented Apr 28, 2023

Copy link
Copy Markdown

@rawagner: This pull request references MGMT-12656 which is a valid jira issue.

Details

In response to this:

This will work only on OCP 4.13 as we need latest pf/react-table

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Apr 28, 2023
@openshift-merge-robot openshift-merge-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label May 4, 2023
@openshift-merge-robot

Copy link
Copy Markdown

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@jtomasek
jtomasek requested a review from celdrake May 5, 2023 07:34
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 5, 2023
@openshift-ci

openshift-ci Bot commented May 5, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: jtomasek, rawagner

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@rawagner

rawagner commented May 5, 2023

Copy link
Copy Markdown
Member Author

already fixed in master by #1919. Closing

@rawagner rawagner closed this May 5, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants