[release-4.20] submodule update 08 19 - #4504
Conversation
Update to openshift/ovn-kubernetes@c28f766 This commit was generated using hack/update_submodules.sh
Update to openshift/prometheus-community-windows_exporter@300b751 This commit was generated using hack/update_submodules.sh
This commit was generated using hack/update_submodules.sh
|
Skipping CI for Draft Pull Request. |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: wgahnagl The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
generated by ./hack/update_submodules.sh