[release-4.21] submodule update 08 19 - #4505
Conversation
|
Skipping CI for Draft Pull Request. |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: wgahnagl The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
| module github.com/openshift/windows-machine-config-operator | ||
|
|
||
| go 1.24.13 | ||
| go 1.25.0 |
There was a problem hiding this comment.
does 4.21 supports go 1.25?
1156ae3 to
60da4e6
Compare
|
@wgahnagl: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
@wgahnagl drop the hybrid-overlay-node commit images — Job failed. |
Update to openshift/prometheus-community-windows_exporter@c0f9d9d This commit was generated using hack/update_submodules.sh
Update to openshift/kubernetes@7ffd39d This commit was generated using hack/update_submodules.sh
This commit was generated using hack/update_submodules.sh
60da4e6 to
213c07a
Compare
generated by ./hack/update_submodules.sh