Skip to content

feat(themes-provider): use static css variables - #6857

Closed
philibea wants to merge 1 commit into
mainfrom
themes-css
Closed

feat(themes-provider): use static css variables#6857
philibea wants to merge 1 commit into
mainfrom
themes-css

Conversation

@philibea

Copy link
Copy Markdown
Collaborator

Summary

Type

  • Bug
  • Feature
  • Enhancement
  • Documentation
  • Migration
  • Refactor

Summarize concisely:

What is expected?

(Description of the new behavior)

The following changes were made:

(Describe what you did)

Relevant logs and/or screenshots

Page Before After
url screenshot screenshot
url screenshot screenshot

@changeset-bot

changeset-bot Bot commented Aug 31, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ac16989

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 6 packages
Name Type
@ultraviolet/themes Minor
@ultraviolet/form Patch
@ultraviolet/icons Patch
@ultraviolet/illustrations Patch
@ultraviolet/nextjs Patch
@ultraviolet/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Signed-off-by: Alexandre Philibeaux <aphilibeaux@scaleway.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants