chore(deps): update dependency @modern-js/plugin-testing to v2.70.4 - #18
chore(deps): update dependency @modern-js/plugin-testing to v2.70.4#18renovate[bot] wants to merge 1 commit into
Conversation
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Pro Plus Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
📝 WalkthroughWalkthroughThe change involves an update to the Changes
Sequence Diagram(s)(No sequence diagrams generated as the changes do not introduce new features or modifications to control flow.) Poem
✨ Finishing Touches🧪 Generate unit tests (beta)
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
b510dab to
a57729e
Compare
a57729e to
640cda0
Compare
640cda0 to
ff1af93
Compare
ff1af93 to
3423f14
Compare
3423f14 to
db0e780
Compare
db0e780 to
a2f747f
Compare
a2f747f to
6d0dcaa
Compare
6d0dcaa to
8906ee7
Compare
8906ee7 to
4c02535
Compare
4c02535 to
1a086f8
Compare
1a086f8 to
821e28e
Compare
821e28e to
ec1aebe
Compare
ec1aebe to
b4d764a
Compare
b4d764a to
23b7aae
Compare
There was a problem hiding this comment.
Actionable comments posted: 0
Review details
Configuration used: CodeRabbit UI
Review profile: CHILL
Commits
Files that changed from the base of the PR and between e049426 and 23b7aae477bb3fbc24d90555394e417305120b56.
Files ignored due to path filters (1)
pnpm-lock.yamlis excluded by!**/pnpm-lock.yaml
Files selected for processing (1)
- package.json (1 hunks)
Files skipped from review due to trivial changes (1)
- package.json
09066a4 to
4b1272d
Compare
4b1272d to
00b062d
Compare
00b062d to
60ac5e1
Compare
60ac5e1 to
1d34c60
Compare
1d34c60 to
94751b4
Compare
94751b4 to
6bd40a2
Compare
6bd40a2 to
013ebe7
Compare
013ebe7 to
b202ba0
Compare
b202ba0 to
d27203e
Compare
d27203e to
f27f02f
Compare
f27f02f to
e228d6f
Compare
e228d6f to
b7d8913
Compare
b7d8913 to
fd584ce
Compare
fd584ce to
14533c7
Compare
This PR contains the following updates:
2.54.0→2.70.4Release Notes
web-infra-dev/modern.js (@modern-js/plugin-testing)
v2.70.4Compare Source
What's Changed
Performance Improvements ⚡
Bug Fixes 🐞
更新内容
性能优化 ⚡
Bug 修复 🐞
v2.70.3Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.70.2Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.70.1Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.70.0Compare Source
What's Changed
New Features 🎉
更新内容
新特性 🎉
v2.69.7Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.69.6Compare Source
What's Changed
New Features 🎉
urlPathfield to Hono contextroute, for later middleware use in #8048Bug Fixes 🐞
Other Changes
更新内容
新特性 🎉
route中添加urlPath字段,以供后续中间件使用, 详情可查看 #8048Bug 修复 🐞
其他变更
v2.69.5Compare Source
What's Changed
Other Changes
更新内容
其他变更
v2.69.4Compare Source
What's Changed
Bug Fixes 🐞
Other Changes
更新内容
Bug 修复 🐞
其他变更
v2.69.3Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.69.2Compare Source
What's Changed
Performance Improvements ⚡
Bug Fixes 🐞
更新内容
性能优化 ⚡
Bug 修复 🐞
v2.69.1Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.69.0Compare Source
What's Changed
New Features 🎉
更新内容
新特性 🎉
v2.68.20Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.19Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.18Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
更新内容
新特性 🎉
Bug 修复 🐞
v2.68.17Compare Source
What's Changed
Bug Fixes 🐞
Other Changes
更新内容
Bug 修复 🐞
其他变更
v2.68.16Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.15Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.14Compare Source
What's Changed
Other Changes
v2.68.13Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
更新内容
新特性 🎉
Bug 修复 🐞
v2.68.12Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.11Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.10Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.9Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.8Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.7Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
更新内容
新特性 🎉
Bug 修复 🐞
v2.68.6Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
Other Changes
更新内容
新特性 🎉
Bug 修复 🐞
v2.68.5Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.68.4Compare Source
What's Changed
New Features 🎉
更新内容
新特性 🎉
v2.68.3Compare Source
What's Changed
Other Changes
更新内容
其他变更
v2.68.2Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
更新内容
新特性 🎉
Bug 修复 🐞
v2.68.1Compare Source
What's Changed
New Features 🎉
modern inspectby @9aoy in #7309更新内容
新特性 🎉
modern inspect命令调用时,支持输出 Modern.js 配置 由 @9aoy 实现, 详情可查看 #7309v2.68.0Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
Other Changes
source-map-supportin prod-server for code start performance by @await-ovo in #7291更新内容
新特性 🎉
Bug 修复 🐞
其他变更
source-map-support以提升冷启动性能 由 @await-ovo 实现, 详情可查看 #7291v2.67.11Compare Source
What's Changed
Docs update 📄
更新内容
文档更新 📄
v2.67.10Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.67.9Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
更新内容
新特性 🎉
Bug 修复 🐞
v2.67.8Compare Source
What's Changed
Bug Fixes 🐞
更新内容
Bug 修复 🐞
v2.67.7Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
更新内容
新特性 🎉
Bug 修复 🐞
v2.67.6Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
更新内容
新特性 🎉
Bug 修复 🐞
v2.67.5Compare Source
What's Changed
New Features 🎉
更新内容
新特性 🎉
v2.67.4Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
Other Changes
更新内容
新特性 🎉
Bug 修复 🐞
v2.67.3Compare Source
What's Changed
New Features 🎉
Other Changes
更新内容
新特性 🎉
其他变更
v2.67.2Compare Source
What's Changed
New Features 🎉
x-modern-ssr-fallbackheader by @await-ovo in #7045@modern-js/imagepackage to provide optimizedBug Fixes 🐞
Docs update 📄
更新内容
新特性 🎉
x-modern-ssr-fallback中自定义降级原因 由 @await-ovo 实现, 详情可查看 #7045@modern-js/image包以提供优化的Bug 修复 🐞
文档更新 📄
v2.67.1Compare Source
What's Changed
New Features 🎉
Bug Fixes 🐞
Docs update 📄
更新内容
新特性 🎉
Bug 修复 🐞
文档更新 📄
v2.67.0Compare Source
What's Changed
New Features 🎉
src/modern.runtime.tsin #6942Bug Fixes 🐞
Docs update 📄
更新内容
新特性 🎉
src/modern.runtime.ts中, 详情可查看 #6942Bug 修复 🐞
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.