Repository navigation
Update Konflux references - #356
red-hat-konflux[bot] wants to merge 1 commit into
Conversation
|
Hi @red-hat-konflux[bot]. Thanks for your PR. I'm waiting for a validatedpatterns member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
d50e9f9 to
206eb4a
Compare
d8690ec to
871b4d7
Compare
9fae24a to
73d4a6e
Compare
73d4a6e to
5efa181
Compare
bfc415b to
94ecf41
Compare
94ecf41 to
4c7d95e
Compare
34f65cd to
3aa8dea
Compare
3aa8dea to
b9dbc27
Compare
b9dbc27 to
00e4a67
Compare
c39e489 to
9abc6f1
Compare
7300958 to
0891d08
Compare
|
/ok-to-test |
0891d08 to
5869122
Compare
5869122 to
c7e25e2
Compare
c7e25e2 to
430a3f8
Compare
430a3f8 to
65e986d
Compare
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/retest |
65e986d to
afba6c5
Compare
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
afba6c5 to
0bb6f2a
Compare
This PR contains the following updates:
0e022da->09d1d4f2b4000d->00593740.2->0.32ca2cb8->4d452339ca3eb9->7b93901ff20a2f->b99d3776d307bd->ebfb60379abddd->aa6e7aa0.3->0.41c46fdc->44b2e200.2->0.3716d50d->599d8b1Configuration
📅 Schedule: Branch creation - "after 5am on saturday" (UTC), Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
To execute skipped test pipelines write comment
/ok-to-test.This PR has been generated by MintMaker (powered by Renovate Bot).