Skip to content

Update golang.org/x/exp digest to 8b4c13b - #373

Closed
red-hat-konflux[bot] wants to merge 1 commit into
mainfrom
konflux/mintmaker/main/golang.org-x-exp-digest
Closed

red-hat-konflux[bot] wants to merge 1 commit into
mainfrom
konflux/mintmaker/main/golang.org-x-exp-digest

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Aug 17, 2025 •

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
golang.org/x/exp indirect digest 8a7402a -> 8b4c13b

Warning

Some dependencies could not be looked up. Check the warning logs for more information.


Configuration

📅 Schedule: Branch creation - "after 5am on sunday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.

This PR has been generated by MintMaker (powered by Renovate Bot).

@openshift-ci

openshift-ci Bot commented Aug 17, 2025

Copy link
Copy Markdown
Contributor

Hi @red-hat-konflux[bot]. Thanks for your PR.

I'm waiting for a validatedpatterns member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/main/golang.org-x-exp-digest branch from 71eb8be to 17f51a5 Compare August 24, 2025 08:59
@red-hat-konflux red-hat-konflux Bot changed the title chore(deps): update golang.org/x/exp digest to 42675ad Update golang.org/x/exp digest to 8b4c13b Aug 24, 2025
@red-hat-konflux

Copy link
Copy Markdown
Contributor Author

ℹ Artifact update notice

File name: go.mod

In order to perform the update(s) described in the table above, Renovate ran the go get command, which resulted in the following additional change(s):

  • 7 additional dependencies were updated
  • The go directive was updated for compatibility reasons

Details:

Package Change
go 1.23.0 -> 1.24.0
go (toolchain) 1.23.2 -> 1.24.6
golang.org/x/crypto v0.37.0 -> v0.41.0
golang.org/x/net v0.39.0 -> v0.43.0
golang.org/x/sync v0.13.0 -> v0.16.0
golang.org/x/sys v0.32.0 -> v0.35.0
golang.org/x/term v0.31.0 -> v0.34.0
golang.org/x/text v0.24.0 -> v0.28.0
golang.org/x/tools v0.31.0 -> v0.36.0

@openshift-ci

openshift-ci Bot commented Aug 24, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: red-hat-konflux[bot]
Once this PR has been reviewed and has the lgtm label, please assign mlabonte-rh for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@mbaldessari mbaldessari closed this Sep 4, 2025
@red-hat-konflux
red-hat-konflux Bot deleted the konflux/mintmaker/main/golang.org-x-exp-digest branch October 8, 2025 16:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant