Skip to content

library/util_pack: Use m_axis_ready to reset accum - #2118

Open
podgori wants to merge 1 commit into
mainfrom
cpack_fix
Open

podgori wants to merge 1 commit into
mainfrom
cpack_fix

Conversation

@podgori

@podgori podgori commented Aug 4, 2026

Copy link
Copy Markdown
Contributor

PR Description

This commit restores the idle behavior and splits it from the backpressure state which discards any partial accumulation, so that the next enabled window starts on a clean boundary.

This behavior was observed in a simulation which toggles CPACK's fifo_wr_en input; in a JESD TPL - CPACK HW chain, the fifo_wr_en input stays asserted since the jesd transfer is continuous, basically combining idle/flush when backpressured.

PR Type

  • Bug fix (change that fixes an issue)
  • New feature (change that adds new functionality)
  • Breaking change (has dependencies in other repos or will cause CI to fail)
  • Documentation

PR Checklist

  • I have followed the code style guidelines
  • I have performed a self-review of changes
  • I have compiled all hdl projects and libraries affected by this PR
  • I have tested in hardware affected projects, at least on relevant boards
  • I have commented my code, at least hard-to-understand parts
  • I have signed off all commits from this PR
  • I have updated the documentation (wiki pages, ReadMe files, Copyright etc)
  • I have not introduced new Warnings/Critical Warnings on compilation
  • I have added new hdl testbenches or updated existing ones

Signed-off-by: Ionut Podgoreanu <ionut.podgoreanu@analog.com>

@IstvanZsSzekely IstvanZsSzekely left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Both unit and system level testbenches passed.

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants