Skip to content

MGMT-12731: Opening the Actions menu in a host table causes a scrollbar to appear - #1919

Merged
jkilzi merged 2 commits into
openshift-assisted:masterfrom
jgyselov:actions_menu_append_to
May 4, 2023
Merged

MGMT-12731: Opening the Actions menu in a host table causes a scrollbar to appear#1919
jkilzi merged 2 commits into
openshift-assisted:masterfrom
jgyselov:actions_menu_append_to

Conversation

@jgyselov

@jgyselov jgyselov commented Feb 14, 2023

Copy link
Copy Markdown
Contributor

ammont82
ammont82 previously approved these changes Feb 15, 2023
@celdrake
celdrake marked this pull request as draft February 16, 2023 07:59
@celdrake

Copy link
Copy Markdown
Contributor

Moving to Draft since we won't be able to merge the UHC portal part yet.

@jgyselov jgyselov added the OCM label Mar 3, 2023
@jgyselov
jgyselov force-pushed the actions_menu_append_to branch from cc5dab2 to 78c23bb Compare April 24, 2023 09:23
@jgyselov
jgyselov marked this pull request as ready for review April 25, 2023 07:51
@jgyselov
jgyselov force-pushed the actions_menu_append_to branch 2 times, most recently from 24b341d to 2d808a3 Compare April 26, 2023 10:01
@jkilzi
jkilzi force-pushed the master branch 3 times, most recently from a06a1cd to bd7c2f1 Compare April 27, 2023 17:09
@openshift-merge-robot openshift-merge-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Apr 28, 2023
@jgyselov
jgyselov force-pushed the actions_menu_append_to branch from 2d808a3 to bed968c Compare April 28, 2023 09:34
@openshift-merge-robot openshift-merge-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Apr 28, 2023
@openshift-ci openshift-ci Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Apr 28, 2023
@jgyselov
jgyselov force-pushed the actions_menu_append_to branch from bed968c to 6c9801c Compare April 28, 2023 09:39
@jkilzi jkilzi changed the title MGMT-12731 - Opening the Actions menu in a host table causes a scrollbar to appear MGMT-12731: Opening the Actions menu in a host table causes a scrollbar to appear Apr 30, 2023
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 30, 2023
@openshift-ci-robot

openshift-ci-robot commented Apr 30, 2023

Copy link
Copy Markdown

@jgyselov: This pull request references MGMT-12731 which is a valid jira issue.

Details

In response to this:

https://issues.redhat.com/browse/MGMT-12731

UHC-Portal PR: https://gitlab.cee.redhat.com/service/uhc-portal/-/merge_requests/3921
Assisted-ui PR: openshift-assisted/assisted-ui#637

Before
image

After:
image

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 4, 2023
@openshift-ci

openshift-ci Bot commented May 4, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: jgyselov, jkilzi

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label May 4, 2023
@openshift-ci

openshift-ci Bot commented May 4, 2023

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

@jkilzi jkilzi added this to the v2.20 milestone May 4, 2023
@jkilzi
jkilzi merged commit 08fcc0a into openshift-assisted:master May 4, 2023
jtomasek pushed a commit to jtomasek/assisted-installer-ui that referenced this pull request May 9, 2023
Co-authored-by: Jonathan Kilzi <jkilzi@redhat.com>
jtomasek pushed a commit that referenced this pull request May 9, 2023
Co-authored-by: Jonathan Kilzi <jkilzi@redhat.com>
@jgyselov
jgyselov deleted the actions_menu_append_to branch October 2, 2023 13:29
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. OCM size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants