Skip to content

[release-4.22] SPLAT-2865: Add vSphere cloud config modules from 3CMO - #2404

Merged
openshift-merge-bot[bot] merged 6 commits into
openshift:release-4.22from
vr4manta:SPLAT-2865
Sep 9, 2026
Merged

[release-4.22] SPLAT-2865: Add vSphere cloud config modules from 3CMO#2404
openshift-merge-bot[bot] merged 6 commits into
openshift:release-4.22from
vr4manta:SPLAT-2865

Conversation

@vr4manta

@vr4manta vr4manta commented Aug 4, 2026

Copy link
Copy Markdown
Contributor

SPLAT-2865

Changes

  • Migrated 3CMO vSphere cloud config modules to library-go
  • Changed Node struct to not be a pointer in CPIConfig

Notes

This PR is backporting the following PRs in relation to the vSphere multi vCenter day 2 support feature:

@openshift-ci-robot

openshift-ci-robot commented Aug 4, 2026

Copy link
Copy Markdown

@vr4manta: This pull request references SPLAT-2865 which is a valid jira issue.

Details

In response to this:

SPLAT-2865

Changes

  • Migrated 3CMO vSphere cloud config modules to library-go
  • Changed Node struct to not be a pointer in CPIConfig

Notes

This PR is backporting the following PRs in relation to the vSphere multi vCenter day 2 support feature:

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Aug 4, 2026
@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Aug 4, 2026
@coderabbitai

coderabbitai Bot commented Aug 4, 2026

Copy link
Copy Markdown

Important

Review skipped

Auto reviews are disabled on base/target branches other than the default branch.

Please check the settings in the CodeRabbit UI or the .coderabbit.yaml file in this repository. To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Repository: openshift/coderabbit/.coderabbit.yaml

Review profile: CHILL

Plan: Enterprise

Run ID: 193dd6f6-c2bb-4016-a722-c70d1e0270f5

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@openshift-ci

openshift-ci Bot commented Aug 4, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@vr4manta vr4manta changed the title WIP: [release-422] SPLAT-2865: Add vSphere cloud config modules from 3CMO WIP: [release-4.22] SPLAT-2865: Add vSphere cloud config modules from 3CMO Aug 4, 2026
@vr4manta

vr4manta commented Aug 4, 2026

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci

openshift-ci Bot commented Aug 4, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-ci

openshift-ci Bot commented Aug 6, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: This PR was included in a payload test run from openshift/api#2969
trigger 0 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

@vr4manta vr4manta changed the title WIP: [release-4.22] SPLAT-2865: Add vSphere cloud config modules from 3CMO [release-4.22] SPLAT-2865: Add vSphere cloud config modules from 3CMO Sep 3, 2026
@vr4manta
vr4manta marked this pull request as ready for review September 3, 2026 17:50
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 3, 2026
@openshift-ci
openshift-ci Bot requested review from bertinatto and elmiko September 3, 2026 17:50
@vr4manta

vr4manta commented Sep 3, 2026

Copy link
Copy Markdown
Contributor Author

/assign @elmiko

@openshift-ci

openshift-ci Bot commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: This PR was included in a payload test run from openshift/api#2969
trigger 1 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/1d51e210-a858-11f1-8021-2ae6565127f3-0

@openshift-ci

openshift-ci Bot commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: This PR was included in a payload test run from openshift/api#2969
trigger 1 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/fb9027b0-a873-11f1-9d70-57471176ae00-0

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: This PR was included in a payload test run from openshift/api#2969
trigger 2 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial-1of2
  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial-2of2

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/bb3dc9d0-ab78-11f1-89a9-861801216705-0

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: This PR was included in a payload test run from openshift/api#2969
trigger 2 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial-1of2
  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial-2of2

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/1248eda0-ab87-11f1-8f62-cd05816d43e3-0

@bertinatto

Copy link
Copy Markdown
Member

/assign @jsafrane

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: This PR was included in a payload test run from openshift/api#2969
trigger 2 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial-1of2
  • periodic-ci-openshift-release-main-nightly-4.22-e2e-vsphere-ovn-zones-techpreview-serial-2of2

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/c30f5bc0-ab90-11f1-99a2-f02f9eef2847-0

@jsafrane

jsafrane commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

/lgtm
/approve

I can see the cherry-picked commits are the same as in master and someone must have reviewed them there ;-)

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 9, 2026

@bertinatto bertinatto left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@openshift-ci

openshift-ci Bot commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bertinatto, jsafrane, vr4manta

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 9, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 8fc95c6 into openshift:release-4.22 Sep 9, 2026
5 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants